Skip to content

std: Test the exit behavior of abort and abort_immediate - #163909

Open
tgross35 wants to merge 3 commits into
rust-lang:mainfrom
tgross35:test-abort
Open

tgross35 wants to merge 3 commits into
rust-lang:mainfrom
tgross35:test-abort

Conversation

@tgross35

@tgross35 tgross35 commented Oct 6, 2026

Copy link
Copy Markdown
Member

There has been discussion about possibly bringing abort into core
with abort_immediate as a fallback. If this or other changes happen,
we would want to make sure abort still raises SIGABRT. Since we don't
seem to be asserting the exact exit behavior anywhere, add a test here.

@rustbot

rustbot commented Oct 6, 2026

Copy link
Copy Markdown
Collaborator

The run-make-support library was changed

cc @jieyouxu

These commits modify the Cargo.lock file. Unintentional changes to Cargo.lock can be introduced when switching branches and rebasing PRs.

If this was unintentional then you should revert the changes before this PR is merged.
Otherwise, you can ignore this comment.

@rustbot rustbot added A-run-make Area: port run-make Makefiles to rmake.rs S-waiting-on-review Status: Awaiting review from the assignee but also interested parties. T-compiler Relevant to the compiler team, which will review and decide on the PR/issue. labels Oct 6, 2026
@rustbot

rustbot commented Oct 6, 2026

Copy link
Copy Markdown
Collaborator

r? @petrochenkov

rustbot has assigned @petrochenkov.
They will have a look at your PR within the next two weeks and either review your PR or reassign to another reviewer.

Use r? to explicitly pick a reviewer

Why was this reviewer chosen?

The reviewer was selected based on:

  • Owners of files modified in this PR: compiler
  • compiler expanded to 77 candidates
  • Random selection from 19 candidates

@tgross35

tgross35 commented Oct 6, 2026

Copy link
Copy Markdown
Member Author

@bors try jobs=test-x86_64-msvc*,test-various,test-arm-android,test-aarch64-apple-*

@rust-bors

This comment has been minimized.

rust-bors Bot pushed a commit that referenced this pull request Oct 6, 2026
std: Test the exit behavior of `abort` and `abort_immediate`


try-job: test-x86_64-msvc*
try-job: test-various
try-job: test-arm-android
try-job: test-aarch64-apple-*
@rust-log-analyzer

This comment has been minimized.

@rust-log-analyzer

This comment has been minimized.

@tgross35
tgross35 force-pushed the test-abort branch 2 times, most recently from 8fa366a to 8312c9f Compare October 7, 2026 00:43
@rust-log-analyzer

This comment has been minimized.

@rust-bors rust-bors Bot added S-waiting-on-author Status: This is awaiting some action (such as code changes or more information) from the author. and removed S-waiting-on-review Status: Awaiting review from the assignee but also interested parties. labels Oct 7, 2026
@rust-bors

This comment was marked as outdated.

@tgross35
tgross35 force-pushed the test-abort branch 2 times, most recently from de8f110 to 8b5b91d Compare October 7, 2026 02:30
@tgross35

tgross35 commented Oct 7, 2026

Copy link
Copy Markdown
Member Author

@bors try jobs=test-aarch64-apple-1,test-arm-android,test-various,test-x86_64-msvc-1,test-x86_64-mingw-*

@rust-bors

This comment has been minimized.

rust-bors Bot pushed a commit that referenced this pull request Oct 7, 2026
std: Test the exit behavior of `abort` and `abort_immediate`


try-job: test-aarch64-apple-1
try-job: test-arm-android
try-job: test-various
try-job: test-x86_64-msvc-1
try-job: test-x86_64-mingw-*
@rust-log-analyzer

This comment has been minimized.

@rust-bors

rust-bors Bot commented Oct 7, 2026

Copy link
Copy Markdown
Contributor

☀️ Try build successful (CI)
Build commit: 0a1aaed (0a1aaed99b48e3e24541bf6a8460ea3bf879d5bf)
Base parent: 8d1a764 (8d1a76430406c877b35d0b627e7f796dcf0dfeca)

There has been discussion about possibly bringing `abort` into `core`
with `abort_immediate` as a fallback. If this or other changes happen,
we would want to make sure `abort` still raises SIGABRT. Since we don't
seem to be asserting the exact exit behavior anywhere, add a test here.

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

A-run-make Area: port run-make Makefiles to rmake.rs S-waiting-on-author Status: This is awaiting some action (such as code changes or more information) from the author. T-compiler Relevant to the compiler team, which will review and decide on the PR/issue.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants