Skip to content

feat(suite): add support for render sub-command - #319

Merged
poiana merged 2 commits into
falcosecurity:mainfrom
ekoops:ekoops/render
Jul 20, 2026
Merged

poiana merged 2 commits into
falcosecurity:mainfrom
ekoops:ekoops/render

Conversation

@ekoops

@ekoops ekoops commented Jul 20, 2026

Copy link
Copy Markdown
Contributor

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind documentation

/kind tests

/kind feature

If this PR prepares a chart release, uncomment:

/kind chart-release

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area commands

/area pkg

/area events

/area chart

/area automation

What this PR does / why we need it:

This PR adds a new suite render sub-command allowing to render the test(s) YAML description and display it to output. The rendering process instantiates any template present in the YAML description. The order and the way sources are evaluated is identical to the one the suite test command offers. For each YAML description source, a separate YAML document is written to output; each of these documents is annotated with a comment indicating the source name.

Moreover, this PR updates the stale documentation of the suite run command.

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

ekoops added 2 commits July 20, 2026 12:51
Signed-off-by: Leonardo Di Giovanna <leonardodigiovanna1@gmail.com>
The `suite render` sub-command allows to render the test(s) YAML
description and display it to output. The rendering process
instantiates any template present in the YAML description. The order
and the way sources are evaluated is identical to the one the
`suite test` command offers. For each YAML description source, a
separate YAML document is written to output; each of these documents
is annotated with a comment indicating the source name.

Signed-off-by: Leonardo Di Giovanna <leonardodigiovanna1@gmail.com>
@poiana

poiana commented Jul 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ekoops, leogr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana
poiana merged commit 98ada62 into falcosecurity:main Jul 20, 2026
6 checks passed
@ekoops
ekoops deleted the ekoops/render branch July 20, 2026 12:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants