Skip to content

fix(suite/loader): avoid YAML-encoding test's source index - #317

Merged
poiana merged 1 commit into
falcosecurity:mainfrom
ekoops:ekoops/do-not-export-src-idx
Jul 20, 2026
Merged

poiana merged 1 commit into
falcosecurity:mainfrom
ekoops:ekoops/do-not-export-src-idx

Conversation

@ekoops

@ekoops ekoops commented Jul 20, 2026

Copy link
Copy Markdown
Contributor

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind documentation

/kind tests

/kind feature

If this PR prepares a chart release, uncomment:

/kind chart-release

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area commands

/area pkg

/area events

/area chart

/area automation

What this PR does / why we need it:

A test's source index doesn't need to be propagated to any child process through the tests description YAML encoding, as it is already propagated in the baggage.

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

A test's source index doesn't need to be propagated to any child
process through the tests description YAML encoding, as it is already
propagated in the baggage.

Signed-off-by: Leonardo Di Giovanna <leonardodigiovanna1@gmail.com>
@poiana

poiana commented Jul 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ekoops, leogr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana
poiana merged commit 1eae1ce into falcosecurity:main Jul 20, 2026
6 checks passed
@ekoops
ekoops deleted the ekoops/do-not-export-src-idx branch July 20, 2026 09:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants